Skip to content

Update the artifact-releaser-test pipeline. - #1319

Open
drewhli wants to merge 1 commit into
GoogleCloudPlatform:masterfrom
drewhli:arle
Open

drewhli wants to merge 1 commit into
GoogleCloudPlatform:masterfrom
drewhli:arle

Conversation

@drewhli

@drewhli drewhli commented Dec 9, 2024

Copy link
Copy Markdown
Contributor
  • Remove pushing for packages built on OSes we no longer support/build.
  • Remove obsolete images from the pipeline, like the windows container and prerelease images.
  • Fix the artifact registry autopush topic link.

This should make the pipeline green again.

/cc @dorileo @zmarano

@google-oss-prow
google-oss-prow Bot requested review from dorileo and zmarano December 9, 2024 20:36
@google-oss-prow

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: drewhli

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

- Remove obsolete builds for packages
- Remove mirantis and docker images from autopush pipeline
- Remove prelease images from autopush pipeline
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

1 participant